ci(packit): run Koji/Bodhi on production - #545
Conversation
Move Koji/Bodhi automation to the production, so that we are able to test the Forgejo dist-git deployment before it gets officially moved. Signed-off-by: Matej Focko <mfocko@packit.dev>
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 07s |
betulependule
left a comment
There was a problem hiding this comment.
LGTM.
I suppose we'll need to resolve the failing tests in a separate PR before we can merge this. Not sure why there are problems with marshmallow_jsonschema when running packit integration tests in this particular repo. I don't remember seeing this error in ogr when packit integration tests were run.
As it has been discovered in this PR, we are missing dependencies for one of the recent changes (JSON schema generation) in the Packit. This is caused by the fact the rev-dep plan is copy-pasted from packit repo rather than imported, therefore import the whole plan instead of copy-paste. Signed-off-by: Matej Focko <mfocko@packit.dev>
|
❌ pre-commit FAILURE in 2m 07s |
Signed-off-by: Matej Focko <mfocko@packit.dev>
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 09s |
Move Koji/Bodhi automation to the production, so that we are able to test the Forgejo dist-git deployment before it gets officially moved.
TODO: