build(deps): bump the aws-sdk group with 5 updates - #995
Conversation
Bumps the aws-sdk group with 5 updates: | Package | From | To | | --- | --- | --- | | [github.com/aws/aws-sdk-go-v2](https://github.com/aws/aws-sdk-go-v2) | `1.45.1` | `1.47.0` | | [github.com/aws/aws-sdk-go-v2/config](https://github.com/aws/aws-sdk-go-v2) | `1.33.2` | `1.33.4` | | [github.com/aws/aws-sdk-go-v2/credentials](https://github.com/aws/aws-sdk-go-v2) | `1.20.2` | `1.20.4` | | [github.com/aws/aws-sdk-go-v2/service/sts](https://github.com/aws/aws-sdk-go-v2) | `1.48.0` | `1.50.0` | | [github.com/aws/aws-sdk-go-v2/service/ssm](https://github.com/aws/aws-sdk-go-v2) | `1.76.0` | `1.78.0` | Updates `github.com/aws/aws-sdk-go-v2` from 1.45.1 to 1.47.0 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@v1.45.1...v1.47.0) Updates `github.com/aws/aws-sdk-go-v2/config` from 1.33.2 to 1.33.4 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@config/v1.33.2...config/v1.33.4) Updates `github.com/aws/aws-sdk-go-v2/credentials` from 1.20.2 to 1.20.4 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@v1.20.2...service/mq/v1.20.4) Updates `github.com/aws/aws-sdk-go-v2/service/sts` from 1.48.0 to 1.50.0 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@service/s3/v1.48.0...service/s3/v1.50.0) Updates `github.com/aws/aws-sdk-go-v2/service/ssm` from 1.76.0 to 1.78.0 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@service/s3/v1.76.0...service/s3/v1.78.0) --- updated-dependencies: - dependency-name: github.com/aws/aws-sdk-go-v2 dependency-version: 1.47.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: aws-sdk - dependency-name: github.com/aws/aws-sdk-go-v2/config dependency-version: 1.33.4 dependency-type: direct:production update-type: version-update:semver-patch dependency-group: aws-sdk - dependency-name: github.com/aws/aws-sdk-go-v2/credentials dependency-version: 1.20.4 dependency-type: direct:production update-type: version-update:semver-patch dependency-group: aws-sdk - dependency-name: github.com/aws/aws-sdk-go-v2/service/sts dependency-version: 1.50.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: aws-sdk - dependency-name: github.com/aws/aws-sdk-go-v2/service/ssm dependency-version: 1.78.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: aws-sdk ... Signed-off-by: dependabot[bot] <support@github.com>
|
Important Review skippedIgnore keyword(s) in the title. ⛔ Ignored keywords (3)
Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Path: .coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
Hi @dependabot[bot]. Thanks for your PR. I'm waiting for a openshift member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
@dependabot[bot]: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #995 +/- ##
==========================================
- Coverage 56.17% 56.14% -0.03%
==========================================
Files 83 83
Lines 6679 6679
==========================================
- Hits 3752 3750 -2
- Misses 2459 2460 +1
- Partials 468 469 +1 🚀 New features to boost your workflow:
|
Bumps the aws-sdk group with 5 updates:
1.45.11.47.01.33.21.33.41.20.21.20.41.48.01.50.01.76.01.78.0Updates
github.com/aws/aws-sdk-go-v2from 1.45.1 to 1.47.0Commits
b189f38Release 2026-09-090905d5fRegenerated Clients9cc4bc5Update API model126fe2dfix GetObject deadlock (#3554)5126885Remove retry metrics header middleware (#3546)bd8de4dExpand S3 operations that check for an error inside an HTTP 200 response (wav...394ee64Release 2026-09-083c9cb8bRegenerated Clients6f85682Update endpoints modelcc3942bUpdate API modelUpdates
github.com/aws/aws-sdk-go-v2/configfrom 1.33.2 to 1.33.4Commits
b189f38Release 2026-09-090905d5fRegenerated Clients9cc4bc5Update API model126fe2dfix GetObject deadlock (#3554)5126885Remove retry metrics header middleware (#3546)bd8de4dExpand S3 operations that check for an error inside an HTTP 200 response (wav...394ee64Release 2026-09-083c9cb8bRegenerated Clients6f85682Update endpoints modelcc3942bUpdate API modelUpdates
github.com/aws/aws-sdk-go-v2/credentialsfrom 1.20.2 to 1.20.4Commits
0d643a8Release 2023-12-07e38b534Regenerated Clientsf394dafUpdate SDK's smithy-go dependency to v1.19.04fb654cUpdate endpoints model8594ca0Update API modele33a34cfix codegen ci (#2418)57a201cregenerate partitions (#2415)be06f02Merge pull request #2344 from aws/feat-request-compression-270bf410Change config loading logic8c02c46Release 2023-12-06Updates
github.com/aws/aws-sdk-go-v2/service/stsfrom 1.48.0 to 1.50.0Commits
4334b43Release 2024-02-169e29187Regenerated Clientsf672c49Update endpoints model7f93578Update API modelda7bcdafeat: add client config passthrough to waiter opts (#2499)1bb0482Release 2024-02-15cc83a2bRegenerated Clients7f44d99Update endpoints model4280ccbUpdate API modela264562fix awsjson error deserialization to not expect string code (#2489)Updates
github.com/aws/aws-sdk-go-v2/service/ssmfrom 1.76.0 to 1.78.0Commits
c33f3e8Release 2025-02-273e30151Regenerated Clients7d47f9dUpdate endpoints modelc8c63a0Update API modelc50ebd4Reapply "Track credential providers via User-Agent Feature ids" (#3017)ef40d0cRelease 2025-02-262d21bb4Regenerated Clients8d810fdUpdate endpoints modela0538c5Update API model21ec8a2Release 2025-02-25Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore <dependency name> major versionwill close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)@dependabot ignore <dependency name> minor versionwill close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)@dependabot ignore <dependency name>will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)@dependabot unignore <dependency name>will remove all of the ignore conditions of the specified dependency@dependabot unignore <dependency name> <ignore condition>will remove the ignore condition of the specified dependency and ignore conditions